Skip to content

Add Tekton files for acm versions: 5.1 5.2 - #458

Open
acm-cicd-prow-bot wants to merge 1 commit into
mainfrom
add-tekton-files-5.1-5.2
Open

acm-cicd-prow-bot wants to merge 1 commit into
mainfrom
add-tekton-files-5.1-5.2

Conversation

@acm-cicd-prow-bot

Copy link
Copy Markdown
Collaborator

This PR adds Tekton pipeline files for the following versions:

  • acm-51
  • acm-52

Generated from existing acm-51 templates.

@openshift-ci

openshift-ci Bot commented Aug 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: acm-cicd-prow-bot
Once this PR has been reviewed and has the lgtm label, please assign bjoydeep for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@acm-cicd-prow-bot
acm-cicd-prow-bot force-pushed the add-tekton-files-5.1-5.2 branch 29 times, most recently from 2e68b82 to ca22c37 Compare August 23, 2026 07:27
@acm-cicd-prow-bot
acm-cicd-prow-bot force-pushed the add-tekton-files-5.1-5.2 branch 29 times, most recently from d50098a to 192fab2 Compare August 26, 2026 22:08
Signed-off-by: OpenShift CI Robot <noreply@openshift.io>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant