Skip to content

Update golang dependencies - #43

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes-sigs:masterfrom
aojea:netlinl_optional
Jul 20, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes-sigs:masterfrom
aojea:netlinl_optional

Conversation

@aojea

@aojea aojea commented Jul 20, 2026 •

Copy link
Copy Markdown
Contributor

update golang dependencies

@aojea aojea changed the title Netlinl optional Netlink optional Jul 20, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jul 20, 2026
@kubernetes-prow
kubernetes-prow Bot requested a review from danwinship July 20, 2026 20:00
@aojea

aojea commented Jul 20, 2026

Copy link
Copy Markdown
Contributor Author

/assign @danwinship

@aojea
aojea force-pushed the netlinl_optional branch from 05ca5af to 2477e96 Compare July 20, 2026 20:05
@aojea aojea changed the title Netlink optional Update golang dependencies Jul 20, 2026
@danwinship

Copy link
Copy Markdown
Collaborator

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 20, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: aojea, danwinship

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 20, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 2163caa into kubernetes-sigs:master Jul 20, 2026
5 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants