Skip to content

Optimize constraint checks for plans with no constraints - #3189

Open
benjie wants to merge 4 commits into
graphile:nextfrom
benjaie:constraints
Open

benjie wants to merge 4 commits into
graphile:nextfrom
benjaie:constraints

Conversation

@benjie

@benjie benjie commented Sep 22, 2026

Copy link
Copy Markdown
Member

Most plans have no constraints; let's make those checks easier!

@changeset-bot

changeset-bot Bot commented Sep 22, 2026 •

Copy link
Copy Markdown

馃 Changeset detected

Latest commit: 0c73163

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
grafast Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants