Update CODEOWNERS and maintainers.yaml files - #2
riaankleinhans wants to merge 2 commits into
Conversation
Signed-off-by: Riaan Kleinhans <riaankleinhans@gmail.com>
|
I'm a little bit surprised to see a new repo created in the org without consultation with maintainers. That's not our usual process in this project. The initial state that was pushed to the repo without any review contains many errors. I think this should have been introduced via a pull request. For example, @riaankleinhans can we force push to an empty repository and then open a PR introducing all the relevant files including codeowners. Nothing against the idea of |
|
@fitzthum The initial state of the repo was provisioned automatically by The metadata is harvested from public sources (landscape, CLOMonitor, governance files across your org), so it's a best effort and will probably contain errors. It's not meant to be authoritative without the Org owners' review. Next step would be CNCF Staff contacting the Org owners and asking them to verify the metadata that was collected by the We onboarded a bunch of projects at the same time, and @riaankleinhans is in the process of going through each one for verifications of metadata and follow up with the Org owners. force-push a clean repo isn't necessary imo, and it requires a few changes to the For the issue with stale maintainers, as mentinoed above, we rely on the Org goverances file (CODEOWNERS, MAINTAINERS, etc) and the |
|
I don't think the information should be pushed into the repo if it has not been verified. If the idea is to have organization owners review the info, this can be done via a PR. Fixing things retroactively with PRs isn't really a substitute for the standard process of reviewing code before it is merged. I'm not sure how widely this info is propagated yet. Maybe it's not too big of a deal, but if people come to this repo today, they will find inaccurate information. This update to the maintainers is good, but there are other errors. |
|
@fitzthum I verify the maintainer information as that affect the project's access to CNCF services. Updating the .project information would also help the TOC move faster through moving levels evaluation of project information. |
|
@riaankleinhans I appreciate the update to the maintainers. This info looks correct. Do you plan on making a PR to correct the other information? Again, my suggestion as a maintainer is to reset this repo to an appropriate first commit and add all significant changes via pull requests. I don't want to be annoying about this, but that's the process that we follow for all contributions. |
We definitely appreciate it and we realize that this is all with good intentions. But creating a repo without consulting or notifying the project maintainers may not be the best open source practice. Even more so if the repo comes loaded with content that is not reviewed neither. My advice at that point is to reset (push force) that repo to an empty state and send a PR with its initial content. We'll review/approve it straight away. Let us know how we can help. |
https://github.com/confidential-containers/confidential-containers/blob/main/MAINTAINERS